Skip to content

zziplib: convert back to autotools build - #91260

Merged
veprbl merged 1 commit into
NixOS:masterfrom
veprbl:pr/zziplib_no_cmake
Jun 24, 2020
Merged

veprbl merged 1 commit into
NixOS:masterfrom
veprbl:pr/zziplib_no_cmake

Conversation

@veprbl

@veprbl veprbl commented Jun 21, 2020

Copy link
Copy Markdown
Member

The cmake support in the v0.13.x branch is not very good. It heavily
relies on add_custom_command and is fragile.

The specific problem we ran into is that the install phase would not
create some of the symlinks on darwin. This breaks reverse
dependencies (e.g. texlive).

Fixes: 4f701dd ('zziplib: 0.13.69 → 0.13.71')

Motivation for this change

zziplib on darwin after #85147 is missing libzzip.dylib, libzzip.la, but provides libzzip-0.13.0.71.dylib:

# diff -u <(ls -1 old/lib | sort) <(ls -1 new/lib | sort)
--- /dev/fd/11  2020-06-21 18:33:23.000000000 -0400
+++ /dev/fd/12  2020-06-21 18:33:23.000000000 -0400
@@ -1,13 +1,13 @@
+libzzip-0.13.0.71.dylib
 libzzip-0.13.dylib
-libzzip.dylib
-libzzip.la
+libzzip-0.dylib
+libzzipfseeko-0.13.0.71.dylib
 libzzipfseeko-0.13.dylib
-libzzipfseeko.dylib
-libzzipfseeko.la
+libzzipfseeko-0.dylib
+libzzipmmapped-0.13.0.71.dylib
 libzzipmmapped-0.13.dylib
-libzzipmmapped.dylib
-libzzipmmapped.la
+libzzipmmapped-0.dylib
+libzzipwrap-0.13.0.71.dylib
 libzzipwrap-0.13.dylib
-libzzipwrap.dylib
-libzzipwrap.la
+libzzipwrap-0.dylib
 pkgconfig

This PR unbreaks texlive.bin.core-big on darwin.

Things done
  • Tested using sandboxing (nix.useSandbox on NixOS, or option sandbox in nix.conf on non-NixOS linux)
  • Built on platform(s)
    • NixOS
    • macOS
    • other Linux distributions
  • Tested via one or more NixOS test(s) if existing and applicable for the change (look inside nixos/tests)
  • Tested compilation of all pkgs that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review wip"
  • Tested execution of all binary files (usually in ./result/bin/)
  • Determined the impact on package closure size (by running nix path-info -S before and after)
  • Ensured that relevant documentation is up to date
  • Fits CONTRIBUTING.md.

The cmake support in the v0.13.x branch is not very good. It heavily
relies on add_custom_command and is fragile.

The specific problem we ran into is that the install phase would not
create some of the symlinks on darwin. This breaks reverse
dependencies (e.g. texlive).

Fixes: 4f701dd ('zziplib: 0.13.69 → 0.13.71')
@veprbl
veprbl requested a review from jtojnar June 21, 2020 23:02
@jtojnar

jtojnar commented Jun 21, 2020

Copy link
Copy Markdown
Member

See also #91241

@ofborg ofborg Bot added 10.rebuild-darwin: 11-100 This PR causes between 11 and 100 packages to rebuild on Darwin. 10.rebuild-linux: 101-500 This PR causes between 101 and 500 packages to rebuild on Linux. labels Jun 21, 2020
@veprbl

veprbl commented Jun 21, 2020 •

Copy link
Copy Markdown
Member Author

I don't think we should be working around bugs in the cmake build for the current release branch of zziplib. It would be much simpler to just stick with autotools build that is working now. We could add zziplib_unstable to work with cmake.

@marsam marsam mentioned this pull request Jun 21, 2020
10 tasks
@veprbl

veprbl commented Jun 21, 2020

Copy link
Copy Markdown
Member Author

@GrahamcOfBorg build mpd texlive.bin.core-big

@veprbl

veprbl commented Jun 24, 2020

Copy link
Copy Markdown
Member Author

@jtojnar Do you mind if we merge this? The problem has been reported upstream. It's unfortunate that I have to revert some of your work, but it should be easy enough to reinstate it, once upstream streamlines their cmake build.

@veprbl
veprbl merged commit a9e15a6 into NixOS:master Jun 24, 2020
@veprbl
veprbl deleted the pr/zziplib_no_cmake branch December 1, 2020 16:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

10.rebuild-darwin: 11-100 This PR causes between 11 and 100 packages to rebuild on Darwin. 10.rebuild-linux: 101-500 This PR causes between 101 and 500 packages to rebuild on Linux.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants