Promote scalars to constants in ndd arithmetic operators - #6475
Open
rostan-t wants to merge 2 commits into
Open
Conversation
Signed-off-by: Rostan Tabet <rtabet@nvidia.com>
Signed-off-by: Rostan Tabet <rtabet@nvidia.com>
Contributor
18 tasks
Collaborator
Author
|
!build |
Collaborator
|
CI MESSAGE: [66962649]: BUILD STARTED |
Collaborator
|
CI MESSAGE: [66962649]: BUILD PASSED |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Category:
Bug fix (non-breaking change which fixes an issue)
Description:
Currently, integers and floats are not promoted to constants in ndd arithmetic operators. This PR fixes it.
With
aandbtwo 64x224x224x3 batches of uint8 images on the GPU,0.4 * a + 0.6 * btakes around 640 us to run on main and below 230 us on this branch.One important detail is that all bool/int/floats are currently promoted to constants, forcing the creation of a new operator instance each time. PR #6476 addresses this issue.
Additional information:
Affected modules and functionalities:
Airthmetic ops
Key points relevant for the review:
Tests:
test_arithm_ops.pyChecklist
Documentation
DALI team only
Requirements
REQ IDs: N/A
JIRA TASK: N/A