Skip to content

Add Nexus to CODEOWNERS - #517

Merged
atol merged 1 commit into
mainfrom
alicelin/nexus-codeowners
Sep 9, 2026
Merged

Add Nexus to CODEOWNERS#517
atol merged 1 commit into
mainfrom
alicelin/nexus-codeowners

Conversation

@atol

@atol atol commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

What was changed

Add the Nexus team as codeowners for samples and tests that start with "nexus". Also, add the Nexus team as co-owners of the README and pnpm-lock.yaml.

@atol
atol requested a review from a team as a code owner September 9, 2026 17:26
@atol
atol merged commit 66ff36e into main Sep 9, 2026
13 checks passed
@atol
atol deleted the alicelin/nexus-codeowners branch September 9, 2026 19:20
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants