Skip to content

[release-4.19] OCPBUGS-122228: Update message for MachineOSBuildFailed condition - #6527

Open
isabella-janssen wants to merge 1 commit into
openshift:release-4.19from
isabella-janssen:ocpbugs-112465-419
Open

isabella-janssen wants to merge 1 commit into
openshift:release-4.19from
isabella-janssen:ocpbugs-112465-419

Conversation

@isabella-janssen

@isabella-janssen isabella-janssen commented Sep 10, 2026

Copy link
Copy Markdown
Member

Closes: OCPBUGS-122228

- What I did
This is a manual backport for #6518.

- How to verify it

  1. Launch a 4.19 cluster with this PR included.
  2. Enable OCL in a pool with a MachineOSConfig that will cause a MachineOSBuild failure. I do this by setting docker-password to an invalid value in the following flow.
$ oc create secret docker-registry quay-push-secret \
    -n openshift-machine-config-operator \
    --docker-server=quay.io \
    --docker-username=<>
    --docker-password=<>
$ oc create -f - << EOF
apiVersion: machineconfiguration.openshift.io/v1
kind: MachineConfigPool
metadata:
 name: layered
spec:
 machineConfigSelector:
  matchExpressions:
   - {key: machineconfiguration.openshift.io/role, operator: In, values: [worker,layered]}
 nodeSelector:
  matchLabels:
   node-role.kubernetes.io/layered: ""
EOF
$ oc label node <node> node-role.kubernetes.io/layered=
$ oc apply -f - << EOF
apiVersion: machineconfiguration.openshift.io/v1
kind: MachineOSConfig
metadata:
  name: layered
spec:
  machineConfigPool:
    name: layered
  containerFile:
  - containerfileArch: NoArch
    content: |-
      FROM configs AS final
      RUN dnf install -y nmap && \
        dnf clean all && \
        bootc container lint
  imageBuilder:
    imageBuilderType: Job
  renderedImagePushSecret:
    name: quay-push-secret
  renderedImagePushSpec: "quay.io/<repo-path>:ocl-testing"
EOF
  1. Once the MachineOSBuild fails, check that the Message field for the Failed condition is being set to a clear value.
$ oc describe machineosbuild layered-15f925a3dabd6139c709d8bb54ffb0bc
Name:         layered-15f925a3dabd6139c709d8bb54ffb0bc
...
Status:
...
  Conditions:
...
    Last Transition Time:  2026-09-10T17:24:56Z
    Message:               Job "build-layered-15f925a3dabd6139c709d8bb54ffb0bc" failed after 4 attempt(s): Job has reached the specified backoff limit
    Reason:                BackoffLimitExceeded
    Status:                True
    Type:                  Failed

- Description for the changelog
OCPBUGS-122228: Update message for MachineOSBuildFailed condition

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 10, 2026
@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@coderabbitai

coderabbitai Bot commented Sep 10, 2026

Copy link
Copy Markdown

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository: openshift/coderabbit/.coderabbit.yaml

Review profile: CHILL

Plan: Enterprise

Run ID: 87392f16-07f6-4750-b3d6-a9604e632315

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@isabella-janssen

Copy link
Copy Markdown
Member Author

/jira cherrypick OCPBUGS-122109

@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: isabella-janssen

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 10, 2026
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@isabella-janssen: Jira Issue OCPBUGS-122109 has been cloned as Jira Issue OCPBUGS-122228. Will retitle bug to link to clone.
/retitle OCPBUGS-122228: Update message for MachineOSBuildFailed condition

Details

In response to this:

/jira cherrypick OCPBUGS-122109

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot changed the title Update message for MachineOSBuildFailed condition OCPBUGS-122228: Update message for MachineOSBuildFailed condition Sep 10, 2026
@openshift-ci-robot openshift-ci-robot added jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Sep 10, 2026
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@isabella-janssen: This pull request references Jira Issue OCPBUGS-122228, which is invalid:

  • expected dependent Jira Issue OCPBUGS-122109 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Closes:

- What I did
This is a manual backport for #6518.

- How to verify it

  1. Launch a 4.19 cluster with this PR included.
  2. Enable OCL in a pool with a MachineOSConfig that will cause a MachineOSBuild failure. I do this by setting docker-password to an invalid value in the following flow.
$ oc create secret docker-registry quay-push-secret \
   -n openshift-machine-config-operator \
   --docker-server=quay.io \
   --docker-username=<>
   --docker-password=<>
$ oc create -f - << EOF
apiVersion: machineconfiguration.openshift.io/v1
kind: MachineConfigPool
metadata:
name: layered
spec:
machineConfigSelector:
 matchExpressions:
  - {key: machineconfiguration.openshift.io/role, operator: In, values: [worker,layered]}
nodeSelector:
 matchLabels:
  node-role.kubernetes.io/layered: ""
EOF
$ oc label node <node> node-role.kubernetes.io/layered=
$ oc apply -f - << EOF
apiVersion: machineconfiguration.openshift.io/v1
kind: MachineOSConfig
metadata:
 name: layered
spec:
 machineConfigPool:
   name: layered
 containerFile:
 - containerfileArch: NoArch
   content: |-
     FROM configs AS final
     RUN dnf install -y nmap && \
       dnf clean all && \
       bootc container lint
 imageBuilder:
   imageBuilderType: Job
 renderedImagePushSecret:
   name: quay-push-secret
 renderedImagePushSpec: "quay.io/<repo-path>:ocl-testing"
EOF
  1. Once the MachineOSBuild fails, check that the Message field for the Failed condition is being set to a clear value.

- Description for the changelog
: Update message for MachineOSBuildFailed condition

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@isabella-janssen isabella-janssen changed the title OCPBUGS-122228: Update message for MachineOSBuildFailed condition [release-4.19] OCPBUGS-122228: Update message for MachineOSBuildFailed condition Sep 10, 2026
@isabella-janssen
isabella-janssen marked this pull request as ready for review September 10, 2026 17:29
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 10, 2026
@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

@isabella-janssen: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@isabella-janssen

Copy link
Copy Markdown
Member Author

/jira refresh

@openshift-ci-robot openshift-ci-robot added jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. and removed jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Sep 14, 2026
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@isabella-janssen: This pull request references Jira Issue OCPBUGS-122228, which is valid. The bug has been moved to the POST state.

7 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.19.z) matches configured target version for branch (4.19.z)
  • bug is in the state New, which is one of the valid states (NEW, ASSIGNED, POST)
  • release note type set to "Release Note Not Required"
  • dependent bug Jira Issue OCPBUGS-122109 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-122109 targets the "4.20.z" version, which is one of the valid target versions: 4.20.0, 4.20.z
  • bug has dependents

No GitHub users were found matching the public email listed for the QA contact in Jira (ptalgulk@redhat.com), skipping review request.

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@isabella-janssen

Copy link
Copy Markdown
Member Author

/verified later @sergiordlr

@openshift-ci-robot openshift-ci-robot added verified-later verified Signifies that the PR passed pre-merge verification criteria labels Sep 16, 2026
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@isabella-janssen: This PR has been marked to be verified later by @sergiordlr.

Details

In response to this:

/verified later @sergiordlr

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. verified Signifies that the PR passed pre-merge verification criteria verified-later

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants