Skip to content

UPSTREAM: <carry>: fix(test/e2e): TestTopologySpreadConstraint: set node selector so pods land to worker nodes only - #1090

Open
ingvagabund wants to merge 1 commit into
openshift:mainfrom
ingvagabund:more-e2e-fixes
Open

UPSTREAM: <carry>: fix(test/e2e): TestTopologySpreadConstraint: set node selector so pods land to worker nodes only#1090
ingvagabund wants to merge 1 commit into
openshift:mainfrom
ingvagabund:more-e2e-fixes

Conversation

@ingvagabund

Copy link
Copy Markdown
Member

ssia

…ode selector so pods land to worker nodes only
@ingvagabund ingvagabund added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. verified Signifies that the PR passed pre-merge verification criteria labels Sep 3, 2026
@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ingvagabund

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 7bed6e2 and 2 for PR HEAD 29ccf44 in total

@openshift-merge-bot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 9c67bf3 and 1 for PR HEAD 29ccf44 in total

@openshift-merge-bot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 774ed02 and 0 for PR HEAD 29ccf44 in total

@ingvagabund

Copy link
Copy Markdown
Member Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 8, 2026
@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown

@ingvagabund: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/security 29ccf44 link false /test security
ci/prow/e2e-aws-operator 29ccf44 link true /test e2e-aws-operator

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant