Skip to content

Auto-request reviews for parts of repository (CODEOWNERS) - #613

Open
lubianat wants to merge 5 commits into
mainfrom
code_owners
Open

lubianat wants to merge 5 commits into
mainfrom
code_owners

Conversation

@lubianat

@lubianat lubianat commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

This adds two files:

A custom action (pinned to a commit) is used because the native mechanism is too rigid (e.g. it requires write permissions to assign someone as a reviewer)

Authors of RFC (surely the corresponding authors) could be auto-notified of changes to their part of the repository.

Additionally, it could allow for ownership of parts of the documentation

"owners" don't get any special right; they just get a review request any time a PR that touches that path is opened.

It uses a GH Action instead of base CODEOWNERS so people without "write" access can be given review-level ownership of parts of the repository.

This allows e.g. auto assigning RFC authors as reviewers for changes that are related to their RFC.
@lubianat lubianat added the infra Concern: basic infrastructure label Sep 15, 2026
@github-actions

Copy link
Copy Markdown
Contributor

Automated Review URLs

@lubianat
lubianat requested a review from jo-mueller September 15, 2026 16:51

This comment was marked as outdated.

joshmoore
joshmoore previously approved these changes Sep 16, 2026

@joshmoore joshmoore left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks, @lubianat! Very happy to try something like this out. FMPOV, I think I'd be happy to be added to rfc/ and specifications/ though being aware of the noise, maybe not :) An alternative for the GerBI team would be to make sure we have all the subdirectories covered, and then we can always add each other.

@lubianat

Copy link
Copy Markdown
Contributor Author

Thank you!

I will study https://github.com/marketplace/actions/auto-request-review as an option for this PR. It looks equally simple and better managed.

I can add you to the two groups if that would reduce noise e.g. ignoring PRs that are pure docs, depending on your notification configs.

Generally: happy to add anyone to any review group upon (pull) request. No special permissions are granted, so it is safe.

@lubianat

Copy link
Copy Markdown
Contributor Author

Sorry for dismissing your review, @joshmoore .

I've changed it to use auto-request-review; it is more popular and more actively maintained, plus seems to have just the right amount of features.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

infra Concern: basic infrastructure

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants