update the caclio registry to capzcicommunity - #590
Conversation
|
This is part of the PR: cb69bf3 |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jrvaldes, linliu-msft-git, marosset, zylxjtu The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
was this broader fix with the defender exclusions proposed in a prev PR? |
|
@zylxjtu: The following test failed, say
Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/test pull-e2e-capz-windows-2025-extension |
|
/hold cancel |
fd2f701
into
kubernetes-sigs:master
|
do we need to backport it to fix this test: https://testgrid.k8s.io/sig-release-1.37-informing#capz-windows-1.37&exclude-non-failed-tests= |
This is to unblock the 1.37 release