Skip to content

trigger 1.30.5 publish - #2437

Merged
istio-testing merged 1 commit into
istio:release-1.30from
PetrMc:petrmc/publish-1.30.5
Sep 21, 2026
Merged

istio-testing merged 1 commit into
istio:release-1.30from
PetrMc:petrmc/publish-1.30.5

Conversation

@PetrMc

@PetrMc PetrMc commented Sep 18, 2026

Copy link
Copy Markdown
Member

coordinated 1.29.8/1.30.5 patch release

@PetrMc
PetrMc requested a review from a team as a code owner September 18, 2026 18:34
@istio-testing istio-testing added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 18, 2026
@PetrMc PetrMc added the do-not-merge/hold Block automatic merging of a PR. label Sep 18, 2026
@istio-testing

Copy link
Copy Markdown
Contributor

@PetrMc: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
publish-warning_release-builder_release-1.30 721bd28 link false /test publish-warning
Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Signed-off-by: Petr McAllister <petr.mcallister@gmail.com>
@PetrMc PetrMc removed the do-not-merge/hold Block automatic merging of a PR. label Sep 21, 2026
@istio-testing
istio-testing merged commit 05d6879 into istio:release-1.30 Sep 21, 2026
5 of 6 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants