Skip to content

[release-0.15] versioning for 0.15.3 - #2103

Merged
openshift-merge-bot[bot] merged 1 commit into
backube:release-0.15from
tesshuflower:0-15-fp-version
Sep 17, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
backube:release-0.15from
tesshuflower:0-15-fp-version

Conversation

@tesshuflower

Copy link
Copy Markdown
Contributor

Describe what this PR does

versioning for downstream rebuilds/fixpacks

Is there anything that requires special attention?

Related issues:

Signed-off-by: Tesshu Flower <tflower@redhat.com>
@openshift-ci

openshift-ci Bot commented Sep 17, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: tesshuflower

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sonarqubecloud

Copy link
Copy Markdown

@falconizmi

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm label Sep 17, 2026
@tesshuflower tesshuflower changed the title [release-0.15]] versioning for 0.15.3 [release-0.15] versioning for 0.15.3 Sep 17, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit ac16c9c into backube:release-0.15 Sep 17, 2026
32 checks passed
@tesshuflower
tesshuflower deleted the 0-15-fp-version branch September 17, 2026 16:27
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants