Skip to content

Markdownlint update v0.18.1 - #2102

Open
tesshuflower wants to merge 2 commits into
backube:mainfrom
tesshuflower:markdownlint-update-v0.18.1
Open

tesshuflower wants to merge 2 commits into
backube:mainfrom
tesshuflower:markdownlint-update-v0.18.1

Conversation

@tesshuflower

Copy link
Copy Markdown
Contributor

Describe what this PR does

  • Updates markdownlint pre-commit check to v0.18.1
  • formatted some existing md files for line length due to stricter markdownlint line-length checks

Is there anything that requires special attention?

Related issues:

renovate-bot and others added 2 commits September 16, 2026 17:14
Signed-off-by: Mend Renovate <bot@renovateapp.com>
Signed-off-by: Tesshu Flower <tflower@redhat.com>
@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: tesshuflower

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sonarqubecloud

Copy link
Copy Markdown

@tesshuflower

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@tesshuflower

Copy link
Copy Markdown
Contributor Author

/retest

@tesshuflower

Copy link
Copy Markdown
Contributor Author

/retest

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants