Skip to content

Update rclone to v1.75.1 - #2098

Merged
openshift-ci[bot] merged 2 commits into
backube:mainfrom
falconizmi:update-rclone-1.75.1
Sep 14, 2026
Merged

openshift-ci[bot] merged 2 commits into
backube:mainfrom
falconizmi:update-rclone-1.75.1

Conversation

@falconizmi

Copy link
Copy Markdown
Contributor

No description provided.

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: falconizmi

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@falconizmi

Copy link
Copy Markdown
Contributor Author

/retest

Signed-off-by: Ismail Ibrahim Quwarah <iquwarah@redhat.com>
Signed-off-by: Ismail Ibrahim Quwarah <iquwarah@redhat.com>
@sonarqubecloud

Copy link
Copy Markdown

@codecov

codecov Bot commented Sep 14, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 68.7%. Comparing base (4f35a2d) to head (405433f).
⚠️ Report is 3 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##            main   #2098     +/-   ##
=======================================
- Coverage   68.8%   68.7%   -0.1%     
=======================================
  Files         58      58             
  Lines       6076    6076             
=======================================
- Hits        4184    4178      -6     
- Misses      1582    1585      +3     
- Partials     310     313      +3     

see 1 file with indirect coverage changes

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@tesshuflower

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm label Sep 14, 2026
@openshift-ci
openshift-ci Bot merged commit a539fc9 into backube:main Sep 14, 2026
76 of 80 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants