Skip to content

add priorityClassName support for controller on helm chart - #2097

Open
JohnMastodon wants to merge 1 commit into
backube:mainfrom
JohnMastodon:helm-priorityclassname
Open

add priorityClassName support for controller on helm chart#2097
JohnMastodon wants to merge 1 commit into
backube:mainfrom
JohnMastodon:helm-priorityclassname

Conversation

@JohnMastodon

Copy link
Copy Markdown

Describe what this PR does
adds priorityClassName support in helm chart for the controller pod

Is there anything that requires special attention?

Related issues:
closes #2095

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: JohnMastodon
Once this PR has been reviewed and has the lgtm label, please assign screeley44 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

Hi @JohnMastodon. Thanks for your PR.

I'm waiting for a backube member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci openshift-ci Bot added the size/S label Sep 8, 2026
@JohnMastodon
JohnMastodon force-pushed the helm-priorityclassname branch from a9fb1dc to 46c3c84 Compare September 9, 2026 00:07
{{- toYaml . | nindent 8 }}
{{- end }}
{{- with .Values.priorityClassName }}
priorityClassName: {{ . }}

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
priorityClassName: {{ . }}
priorityClassName: {{ . | quote }}

Could we quote just to ensure we get a string value?

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@JohnMastodon Thanks for this PR - I just have one small comment above - also, you will need to make sure you sign your commit - see: https://github.com/backube/volsync/pull/2097/checks?check_run_id=102285384157

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@tesshuflower Thanks for the feedback, I included your suggestion in the commit and signed it

Signed-off-by: John Mastodon <121522530+JohnMastodon@users.noreply.github.com>
@JohnMastodon
JohnMastodon force-pushed the helm-priorityclassname branch from 46c3c84 to 69b0033 Compare September 9, 2026 22:50
@sonarqubecloud

sonarqubecloud Bot commented Sep 9, 2026

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

add priorityClassName passthrough to the helm chart

2 participants