Skip to content

Update restic to v0.19.1 - #2094

Merged
openshift-merge-bot[bot] merged 3 commits into
backube:mainfrom
tesshuflower:update-restic-to-v0.19.1
Sep 9, 2026
Merged

Update restic to v0.19.1#2094
openshift-merge-bot[bot] merged 3 commits into
backube:mainfrom
tesshuflower:update-restic-to-v0.19.1

Conversation

@tesshuflower

@tesshuflower tesshuflower commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

Describe what this PR does

  • updates restic to v0.19.1
  • updates entry.sh as restic init has changed slightly and old message we used to detect bucket not existing has changed.
  • also updates changelog to keep it up-to-date

Is there anything that requires special attention?

Related issues:

@tesshuflower

Copy link
Copy Markdown
Contributor Author

/retest

Signed-off-by: Tesshu Flower <tflower@redhat.com>
Signed-off-by: Tesshu Flower <tflower@redhat.com>
Signed-off-by: Tesshu Flower <tflower@redhat.com>
@tesshuflower
tesshuflower force-pushed the update-restic-to-v0.19.1 branch from eb90d47 to 0a1a77c Compare September 8, 2026 19:34
@sonarqubecloud

sonarqubecloud Bot commented Sep 8, 2026

Copy link
Copy Markdown

@tesshuflower

Copy link
Copy Markdown
Contributor Author

/cc @JohnStrunk

@openshift-ci
openshift-ci Bot requested a review from JohnStrunk September 9, 2026 13:01

@JohnStrunk JohnStrunk left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: JohnStrunk, tesshuflower

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [JohnStrunk,tesshuflower]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit b7a73cb into backube:main Sep 9, 2026
61 of 62 checks passed
@tesshuflower
tesshuflower deleted the update-restic-to-v0.19.1 branch September 9, 2026 20:06
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants