Update pre-commit hook markdownlint/markdownlint to v0.17.0 - #2024
Update pre-commit hook markdownlint/markdownlint to v0.17.0#2024renovate-bot wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
a1c91ca to
5b6debd
Compare
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #2024 +/- ##
=====================================
Coverage 68.6% 68.6%
=====================================
Files 58 58
Lines 6056 6056
=====================================
Hits 4158 4158
Misses 1584 1584
Partials 314 314 🚀 New features to boost your workflow:
|
Signed-off-by: Mend Renovate <bot@renovateapp.com>
5b6debd to
d964011
Compare
|
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |



This PR contains the following updates:
v0.15.0→v0.17.0Note: The
pre-commitmanager in Renovate is not supported by thepre-commitmaintainers or community. Please do not report any problems there, instead create a Discussion in the Renovate repository if you have any questions.Release Notes
markdownlint/markdownlint (markdownlint/markdownlint)
v0.17.0Compare Source
Added
treat_links_as_single_wordoption to MD013#580
Fixed
#580
#580
#580
v0.16.0Compare Source
Added
#519
Changed
#516
#539
#551
#554
#556
#557
#561
#569
Fixed
#539
#539
#541
#542
#543
#543
#545
#546
#547
#548
#549
#550
#552
#553
#555
#558
#560
#562
#564
#565
#566
#563
#572
Configuration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.