tests/artifactcache/expiry.py: Tweak tests for buildbox-casd update - #2191
Merged
Merged
Conversation
buildbox-casd 1.4.23+ triggers LRU expiry on a cleanup threshold below the quota to avoid blocking writes during cleanup. The expiry tests were relying on LRU expiry implementation details that no longer hold. This tweaks a few parameters to pass the tests with the latest and older versions of buildbox-casd. https://gitlab.com/BuildGrid/buildbox/buildbox/-/merge_requests/1204
juergbi
requested review from
BenjaminSchubert,
abderrahim,
cs-shadow and
gtristan
as code owners
September 13, 2026 07:34
|
This PR fixes the issue I reported with the Buildbox update in Nixpkgs, and cross-package rebuilds being picked up by the Nixpkgs CI. I am keen for this to be merged, and when ready, a patch release. However, there is no immediate hurry, as the Buildbox package currently in Nixpkgs is functioning. |
abderrahim
approved these changes
Sep 14, 2026
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
buildbox-casd 1.4.23+ triggers LRU expiry on a cleanup threshold below the quota to avoid blocking writes during cleanup.
The expiry tests were relying on LRU expiry implementation details that no longer hold. This tweaks a few parameters to pass the tests with the latest and older versions of buildbox-casd.
https://gitlab.com/BuildGrid/buildbox/buildbox/-/merge_requests/1204