Skip to content

[Feature] Product-based release flow and scripts - #637

Open
art-alexeyenko wants to merge 10 commits into
feature/product-based-releasesfrom
feature/jss-9682-product-release-flow
Open

art-alexeyenko wants to merge 10 commits into
feature/product-based-releasesfrom
feature/jss-9682-product-release-flow

Conversation

@art-alexeyenko

Copy link
Copy Markdown
Collaborator

Part 3 of product-based release rework of the monorepo. The final goals of the initiative are to:

  • be able to release products and packages separately and preserve the option to release all at once (the way it's done now)
  • be able to track legacy (major - 1) versions of each product (nextjs, angular etc) to deliver hotfixes and minor updates when needed

The full plan is:

  • Create separate template packages for product samples
  • Update changesets to version 3
  • Update Content SDK release scripts and GitHub Actions to include support for product releases and legacy product releases (this PR)

This current PR:

  • Adds and enforces changeset split logic. In order to reliably pick changes to publish for a separate product release, we need separate changeset files. yarn changeset automatically splits the changeset. Additionally, PR action is added to do a breakdown during PR stage.
  • Adds the 'Limited Release' action to prepare a product release, or a release for a single package. It prompts users to select packages for release, creates a separate branch and marks non-selected package to be ignored by changesets
  • Adds a trigger for legacy branches creation. Legacy product branch is created when a major release is triggered for a product package. Legacy branch is a snapshot of repo at the time of previous major version.
    • Legacy branches are created for doing hotfixes for older versions easier
  • Changes the 'Hotfix Prep' action to create release/hotfix/xyz branch (updated name)

@changeset-bot

changeset-bot Bot commented Sep 28, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ae3d188

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
create-content-sdk-app Patch
@sitecore-content-sdk/nextjs Minor
@sitecore-content-sdk/angular Patch
@sitecore-content-sdk/content Minor
@sitecore-content-sdk/react Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@art-alexeyenko art-alexeyenko changed the title Feature/jss 9682 product release flow [Feature] Product-based release flow and scripts Sep 28, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant