Skip to content

thunderbird-bin: throw on unsupported system - #453333

Merged
SuperSandro2000 merged 1 commit into
NixOS:masterfrom
wegank:thunderbird-bin-throw-system
Oct 19, 2025
Merged

SuperSandro2000 merged 1 commit into
NixOS:masterfrom
wegank:thunderbird-bin-throw-system

Conversation

@wegank

@wegank wegank commented Oct 18, 2025 •

Copy link
Copy Markdown
Member

#421125 (comment)

It's probably the only thing to fix on aarch64-linux, as there's still a large user base using nixpkgs-review there. For exotic platforms, sorry — just try blast-bin. See #420442 for the current situation on riscv64-linux.

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

Add a 👍 reaction to pull requests you find important.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 0 This PR does not cause any packages to rebuild on Linux. 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Oct 18, 2025
@h0nIg h0nIg mentioned this pull request Oct 19, 2025
2 of 13 tasks
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. and removed 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Oct 19, 2025
@SuperSandro2000
SuperSandro2000 added this pull request to the merge queue Oct 19, 2025
Merged via the queue into NixOS:master with commit b2c89ef Oct 19, 2025
32 of 35 checks passed
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 2 This PR was reviewed and approved by two persons. and removed 12.approvals: 1 This PR was reviewed and approved by one person. labels Oct 19, 2025
@h0nIg h0nIg mentioned this pull request Oct 19, 2025
5 of 13 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 0 This PR does not cause any packages to rebuild on Linux. 12.approvals: 2 This PR was reviewed and approved by two persons.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants