extract-dtb: init at 1.2.3 - #423755
Merged
Merged
extract-dtb: init at 1.2.3#423755
Conversation
iamanaws
reviewed
Jul 10, 2025
ungeskriptet
force-pushed
the
extract-dtb
branch
from
July 10, 2025 04:18
31091fb to
783393c
Compare
isabelroses
reviewed
Jul 10, 2025
isabelroses
reviewed
Jul 10, 2025
ungeskriptet
force-pushed
the
extract-dtb
branch
from
July 10, 2025 10:16
783393c to
c5895f8
Compare
ungeskriptet
force-pushed
the
extract-dtb
branch
from
July 28, 2025 15:18
c5895f8 to
69ddebb
Compare
Member
|
iamanaws
approved these changes
Jul 31, 2025
isabelroses
approved these changes
Jul 31, 2025
Contributor
|
one who use extract-dtb must have interests on this pr: a nixos-module that support out-of-tree boards without patching kernel source, see #403746 |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Add extract-dtb, a program to split appended DTBs from kernel images or DTBOs.
Depends on #416737 for the maintainer list update.Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.