netbird: remove misuzu as maintainer - #341812
Merged
2 commits merged intoSep 15, 2024
Merged
2 commits merged into
2 commits merged into
Conversation
Aleksanaa
approved these changes
Sep 14, 2024
Member
|
@vrifox would you like to maintain the NixOS module and test? |
|
@Aleksanaa I'm fairly new to maintaining packages, but I will do my best, yes. :) |
Contributor
|
Backport failed for Please cherry-pick the changes locally and resolve any conflicts. git fetch origin release-24.05
git worktree add -d .worktree/backport-341812-to-release-24.05 origin/release-24.05
cd .worktree/backport-341812-to-release-24.05
git switch --create backport-341812-to-release-24.05
git cherry-pick -x ad3c462a6ecdabfaa0ae4f5d694ba3f299247b96 2505777e0c90e4ae553b40033dfd931aaa442ea5 |
Member
|
I'm actually driving Netbird daily with no plan to stop in foreseeable future, but I wouldn't really be able to maintain it until my PR is merged |
This pull request was closed.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description of changes
I've migrated to tailscale and I'm not able to reliably test this package anymore
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.