Skip to content

ruby: Fix withPackages on darwin with makeBinaryWrapper - #210228

Merged
marsam merged 1 commit into
NixOS:masterfrom
amarshall:ruby-makebinarywrapper
Mar 11, 2023
Merged

marsam merged 1 commit into
NixOS:masterfrom
amarshall:ruby-makebinarywrapper

Conversation

@amarshall

Copy link
Copy Markdown
Member
Description of changes

See also #161298

Things done
  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandbox = true set in nix.conf? (See Nix manual)
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • 23.05 Release Notes (or backporting 22.11 Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
    • (Release notes changes) Ran nixos/doc/manual/md-to-db.sh to update generated release notes
  • Fits CONTRIBUTING.md.

nixpkgs-review result

$ git -c fetch.prune=false fetch --no-tags --force https://github.com/NixOS/nixpkgs master:refs/nixpkgs-review/0
From https://github.com/NixOS/nixpkgs
   c07552f6f7d..851558501d2  master     -> refs/nixpkgs-review/0
$ git worktree add /home/user/.cache/nixpkgs-review/rev-baf1f1293b289ce53d9b660465cd68e1a3e00b96/nixpkgs 851558501d28bd3117010f71cd1afc7684394849
Preparing worktree (detached HEAD 851558501d2)
Updating files: 100% (33158/33158), done.
HEAD is now at 851558501d2 Merge pull request #210162 from r-ryantm/auto-update/cargo-geiger
$ nix-env --option system x86_64-linux -f /home/user/.cache/nixpkgs-review/rev-baf1f1293b289ce53d9b660465cd68e1a3e00b96/nixpkgs -qaP --xml --out-path --show-trace
$ git merge --no-commit --no-ff baf1f1293b289ce53d9b660465cd68e1a3e00b96
Automatic merge went well; stopped before committing as requested
When finished, apply stashed changes with `git stash pop`
$ nix-env --option system x86_64-linux -f /home/user/.cache/nixpkgs-review/rev-baf1f1293b289ce53d9b660465cd68e1a3e00b96/nixpkgs -qaP --xml --out-path --show-trace --meta
2 packages updated:
snmpcheck taoup

$ nix --experimental-features nix-command build --no-link --keep-going --option build-use-sandbox relaxed -f /home/user/.cache/nixpkgs-review/rev-baf1f1293b289ce53d9b660465cd68e1a3e00b96/build.nix
2 packages built:
snmpcheck taoup

@amarshall
amarshall requested a review from marsam as a code owner January 11, 2023 16:33
@github-actions github-actions Bot added the 6.topic: ruby A dynamic, open source programming language with a focus on simplicity and productivity. label Jan 11, 2023
@ofborg ofborg Bot added 6.topic: darwin Running or building packages on Darwin 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. labels Jan 11, 2023
@ncfavier

Copy link
Copy Markdown
Member

Looks good. Could we add pkgs/development/ruby-modules/with-packages/test.nix to ruby.passthru.tests or something?

@mweinelt mweinelt added the 12.approvals: 1 This PR was reviewed and approved by one person. label Feb 1, 2023
@marsam
marsam merged commit 8b6900d into NixOS:master Mar 11, 2023
@drupol drupol mentioned this pull request Mar 30, 2023
12 tasks
@amarshall
amarshall deleted the ruby-makebinarywrapper branch June 12, 2023 13:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

6.topic: darwin Running or building packages on Darwin 6.topic: ruby A dynamic, open source programming language with a focus on simplicity and productivity. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants