chore(deps): update compliance-operator-content-dev to 3bdfe15 - #1384
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @red-hat-konflux-kflux-prd-rh02[bot]. Thanks for your PR. I'm waiting for a ComplianceAsCode member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
🤖 To deploy this PR, run the following command: |
d460226 to
6dc17a2
Compare
|
🤖 To deploy this PR, run the following command: |
6dc17a2 to
f819a76
Compare
|
🤖 To deploy this PR, run the following command: |
f819a76 to
8ee307e
Compare
|
🤖 To deploy this PR, run the following command: |
8ee307e to
aade2ea
Compare
|
🤖 To deploy this PR, run the following command: |
aade2ea to
d1e90f7
Compare
|
🤖 To deploy this PR, run the following command: |
d1e90f7 to
19ae93c
Compare
19ae93c to
302161c
Compare
|
🤖 To deploy this PR, run the following command: |
302161c to
4595140
Compare
0c771fc to
56583c1
Compare
|
🤖 To deploy this PR, run the following command: |
56583c1 to
77f4a04
Compare
77f4a04 to
97857a5
Compare
|
🤖 To deploy this PR, run the following command: |
|
🤖 To deploy this PR, run the following command: |
97857a5 to
612cfa0
Compare
|
🤖 To deploy this PR, run the following command: |
612cfa0 to
a42f8fe
Compare
|
🤖 To deploy this PR, run the following command: |
a42f8fe to
3b2e54d
Compare
|
🤖 To deploy this PR, run the following command: |
3b2e54d to
4e767fe
Compare
|
🤖 To deploy this PR, run the following command: |
4e767fe to
909514f
Compare
909514f to
177c077
Compare
|
🤖 To deploy this PR, run the following command: |
Image created from 'https://github.com/ComplianceAsCode/content?rev=022bad6eeaa40b2353c7d38c3677afe8d96d539f' Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
177c077 to
0c6a05b
Compare
|
🤖 To deploy this PR, run the following command: |
|
🤖 To deploy this PR, run the following command: |
Image created from 'https://github.com/ComplianceAsCode/content?rev=022bad6eeaa40b2353c7d38c3677afe8d96d539f'
This PR contains the following updates:
18c3da9->3bdfe15Configuration
📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
To execute skipped test pipelines write comment
/ok-to-testThis is Konflux nudging PR, not Mintmaker.